Skip to content

Update CODEOWNERS file - #2974

Merged
pshriwise merged 1 commit into
openmc-dev:developfrom
paulromano:update-codeowners
Apr 26, 2024
Merged

pshriwise merged 1 commit into
openmc-dev:developfrom
paulromano:update-codeowners

Conversation

@paulromano

Copy link
Copy Markdown
Contributor

Description

A few updates in our CODEOWNERS file:

For those of you I've just tagged, please let me know if you object to being the "owner". It just means that you'll get pinged for a review if someone changes one of the associated files.

Checklist

  • I have performed a self-review of my own code
  • I have run clang-format (version 15) on any C++ source files (if applicable)
  • I have followed the style guidelines for Python source files (if applicable)
  • I have made corresponding changes to the documentation (if applicable)
  • I have added tests that prove my fix is effective or that my feature works (if applicable)

@pshriwise pshriwise left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks like we need to provide additional permissions to @marquezj to be a codeowner:

image

Otherwise this looks good to me! Thanks for updating this 💯

@paulromano

Copy link
Copy Markdown
Contributor Author

I invited @marquezj to the committers team so just waiting on him to accept 😄

@marquezj

Copy link
Copy Markdown
Contributor

I invited @marquezj to the committers team so just waiting on him to accept 😄

Great! Yes, I accept.

Should I do something on my side?

@paulromano

Copy link
Copy Markdown
Contributor Author

@marquezj Nope! You've done everything you need to do. Thanks for accepting!

@ebknudsen

Copy link
Copy Markdown
Contributor

Just to be clear - I gladly accept my task as well.

@drewejohnson

Copy link
Copy Markdown
Contributor

I don't mind the nagging, just so long as you all don't mind me being very slow to respond 😅 I'm not on GitHub that much, but I still want to support this project as I can

@pshriwise pshriwise left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks @paulromano!

@pshriwise
pshriwise merged commit e8ae706 into openmc-dev:develop Apr 26, 2024
@paulromano
paulromano deleted the update-codeowners branch April 27, 2024 14:08
church89 pushed a commit to openmsr/openmc that referenced this pull request Jul 18, 2024
apingegno pushed a commit to apingegno/openmc that referenced this pull request May 7, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants