Skip to content

Automator: update envoy@ in istio/proxy@release-1.28 - #7323

Open
istio-testing wants to merge 1 commit into
istio:release-1.28from
istio-testing:none-release-1.28-update_envoy_dep-f8560674
Open

istio-testing wants to merge 1 commit into
istio:release-1.28from
istio-testing:none-release-1.28-update_envoy_dep-f8560674

Conversation

@istio-testing

@istio-testing istio-testing commented Oct 1, 2026 •

Copy link
Copy Markdown
Collaborator

Generated by Automator - 2026-10-02T21:58:35+00:00

@istio-testing istio-testing added auto-merge Set this label on a PR to auto-merge it on success of presubmit tests size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Oct 1, 2026
@istio-testing
istio-testing force-pushed the none-release-1.28-update_envoy_dep-f8560674 branch from 27ef3e1 to b94ce03 Compare October 2, 2026 21:58
@istio-testing

Copy link
Copy Markdown
Collaborator Author

@istio-testing: The following tests failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
test-arm-arm64_proxy_release-1.28 b94ce03 link true /test test-arm-arm64
release-test-arm64_proxy_release-1.28 b94ce03 link true /test release-test-arm64
release-test_proxy_release-1.28 b94ce03 link true /test release-test
Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

auto-merge Set this label on a PR to auto-merge it on success of presubmit tests size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant