Repository navigation
python310Packages.numba: Backport support for numpy 1.24.x - #218929
Merged
Merged
Conversation
This is a last ditch effort to keep numba working.
Member
|
If it works and upstream does not report any major issues, then it is OK for me. Otherwise, I am of the opinion we should wait with numpy 1.24. |
Member
Author
|
Has been reviewed and they are waiting on some package availability, not sure. |
Member
|
Result of 15 packages marked as broken and skipped:
86 packages failed to build:
4 packages built:
Not very helpful... |
Member
Author
|
Let's roll with it and see if we can make things work. |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This is a last ditch effort to keep numba working.
The patch is 13 kB large, but this is the price we have to pay.
This is numba/numba/pull/8691 sans documentation and ci changes.
Description of changes
Things done
sandbox = trueset innix.conf? (See Nix manual)nix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)