Skip to content

fix: support parsing camel case in container queries - #11131

Merged
trueadm merged 1 commit into
sveltejs:svelte-4from
tanhauhau:tanlh/fix/support-css-container-queries-camelcase
Apr 12, 2024
Merged

trueadm merged 1 commit into
sveltejs:svelte-4from
tanhauhau:tanlh/fix/support-css-container-queries-camelcase

Conversation

@tanhauhau

Copy link
Copy Markdown
Member

Fixes #10845

The eatIdent(...) method from css-tree somehow expects lowercase identifiers only, and thus not working for camel case container names. switching to a more generic eat(Ident) would work. Since we already know the current token is an identifier with the name, we dont have to assert the content of the identifier via eatIdent()

Before submitting the PR, please make sure you do the following

  • It's really useful if your PR references an issue where it is discussed ahead of time. In many cases, features are absent for a reason. For large changes, please create an RFC: https://github.com/sveltejs/rfcs
  • Prefix your PR title with feat:, fix:, chore:, or docs:.
  • This message body should clearly illustrate what problems it solves.
  • Ideally, include a test that fails without this PR but passes with it.

Tests and linting

  • Run the tests with pnpm test and lint the project with pnpm lint

@changeset-bot

changeset-bot Bot commented Apr 11, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 06e7180

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
svelte Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@tanhauhau
tanhauhau force-pushed the tanlh/fix/support-css-container-queries-camelcase branch from e9576c6 to 06e7180 Compare April 11, 2024 14:45

This branch was successfully deployed

1 active deployment
Preview – svelte-5-preview — 06e71807 Deployed Apr 12, 2024 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants